Skip to content

Run Cyborg containers as cyborg user in the controlplane - #1155

Merged
openshift-merge-bot[bot] merged 1 commit into
openstack-k8s-operators:mainfrom
amoralej:cyborg-user
Jul 27, 2026
Merged

openshift-merge-bot[bot] merged 1 commit into
openstack-k8s-operators:mainfrom
amoralej:cyborg-user

Conversation

@amoralej

@amoralej amoralej commented Jul 24, 2026

Copy link
Copy Markdown
Contributor

Currently, it's running as root (uid = 0) needlessly instead of as cyborg user.

This patch sets both the dbsync, conductor and api to run as cyborg user.

Currently, it's running as root (uid = 0) needlessly instead of as
cyborg user.

This patch sets both the dbsync, conductor and api to run as cyborg
user.

Signed-off-by: Alfredo Moralejo <amoralej@redhat.com>
@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/cf785d60769a4279bcad560549b18a6e

openstack-meta-content-provider FAILURE in 11m 07s
⚠️ nova-operator-kuttl SKIPPED Skipped due to failed job openstack-meta-content-provider
⚠️ nova-operator-tempest-multinode SKIPPED Skipped due to failed job openstack-meta-content-provider
⚠️ nova-operator-tempest-multinode-ceph SKIPPED Skipped due to failed job openstack-meta-content-provider

@amoralej

Copy link
Copy Markdown
Contributor Author

check-rdo

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/ffa45f53349d493cb5ca15178190baff

✔️ openstack-meta-content-provider SUCCESS in 3h 01m 46s
✔️ nova-operator-kuttl SUCCESS in 58m 32s
✔️ nova-operator-tempest-multinode SUCCESS in 2h 39m 25s
nova-operator-tempest-multinode-ceph NODE_FAILURE Node(set) request 099-0000151940 failed in 0s

@amoralej

Copy link
Copy Markdown
Contributor Author

check-rdo

@amoralej
amoralej requested a review from mrkisaolamb July 27, 2026 07:34
Comment thread internal/cyborg/constants.go

@amartyasinha amartyasinha left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@openshift-ci

openshift-ci Bot commented Jul 27, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: amartyasinha, amoralej, SeanMooney

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:
  • OWNERS [SeanMooney,amartyasinha,amoralej]

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-merge-bot
openshift-merge-bot Bot merged commit c258a14 into openstack-k8s-operators:main Jul 27, 2026
7 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants