Skip to content

NO-JIRA: Update openshift/api and generate clients - #397

Merged
openshift-merge-bot[bot] merged 2 commits into
openshift:masterfrom
ardaguclu:new-kms-api-fake
Sep 23, 2026
Merged

openshift-merge-bot[bot] merged 2 commits into
openshift:masterfrom
ardaguclu:new-kms-api-fake

Conversation

@ardaguclu

Copy link
Copy Markdown
Member

In order to get openshift/api#3028 into the client-go.

@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Sep 23, 2026
@openshift-ci-robot

Copy link
Copy Markdown

@ardaguclu: This pull request explicitly references no jira issue.

Details

In response to this:

In order to get openshift/api#3028 into the client-go.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci

openshift-ci Bot commented Sep 23, 2026

Copy link
Copy Markdown
Contributor

@ardaguclu: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@tjungblu

Copy link
Copy Markdown

/lgtm
/approve
/verified by me

@openshift-ci-robot openshift-ci-robot added the verified Signifies that the PR passed pre-merge verification criteria label Sep 23, 2026
@openshift-ci-robot

Copy link
Copy Markdown

@tjungblu: This PR has been marked as verified by me.

Details

In response to this:

/lgtm
/approve
/verified by me

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Sep 23, 2026
@JoelSpeed

Copy link
Copy Markdown
Contributor

/approve

@openshift-ci

openshift-ci Bot commented Sep 23, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ardaguclu, JoelSpeed, tjungblu

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 23, 2026
@openshift-merge-bot
openshift-merge-bot Bot merged commit 89e5bba into openshift:master Sep 23, 2026
3 checks passed
@ardaguclu
ardaguclu deleted the new-kms-api-fake branch September 23, 2026 09:34
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged. verified Signifies that the PR passed pre-merge verification criteria

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants