Skip to content

Fix TestAccDataprocCluster_AttachedDiskConfigWorker - #18626

Open
SirGitsalot wants to merge 1 commit into
mainfrom
SirGitsalot-fix-dataproc-cluster-attached-disk-config-worker-test
Open

Fix TestAccDataprocCluster_AttachedDiskConfigWorker#18626
SirGitsalot wants to merge 1 commit into
mainfrom
SirGitsalot-fix-dataproc-cluster-attached-disk-config-worker-test

Conversation

@SirGitsalot

Copy link
Copy Markdown
Member

The main point of this PR is to fix the "Error 400: Permissions are missing for the default service account" error by putting the cluster into the same region as the buckets. Stockouts are still a big problem for the dataproc tests, but switching to c4-standard-2 might help with that.

Fixes hashicorp/terraform-provider-google#28701 (at least it fixes it locally, we'll see what happens on CI) Does not fix the issue mentioned in the comment regarding oneof ... must be specified which I think is a different problem.


The main point of this PR is to fix the "Error 400: Permissions are missing for the default service account" error by putting the cluster into the same region as the buckets. Stockouts are still a big problem for the dataproc tests, but switching to `c4-standard-2` *might* help with that.
@modular-magician

modular-magician commented Aug 13, 2026

Copy link
Copy Markdown
Collaborator

Hi there, I'm the Modular magician. I've detected the following information about your changes for commit 08ebea6:

Diff report

Your PR generated the following diffs in downstream repositories:

Repository Diff Link Changes
google provider View Diff 1 file changed, 17 insertions(+), 10 deletions(-)
google-beta provider View Diff 1 file changed, 17 insertions(+), 10 deletions(-)

Test report

Analytics

Total Tests Passed Skipped Affected
91 84 6 1
Affected Service Packages
  • dataproc

Learn how VCR tests work


Step 1: Replaying Mode

Action taken

Found 1 affected test(s) by replaying old test recordings. Starting RECORDING based on the most recent commit.

Click here to see the affected tests
  • TestAccDataprocCluster_AttachedDiskConfigWorker

View the replaying VCR build log


Step 2: Recording Mode

Recording Mode Replaying Rerun Test Name
✅ Log TestAccDataprocCluster_AttachedDiskConfigWorker

🟢 All tests passed!

View the recording VCR build log or the debug logs folder for detailed results.

@SirGitsalot VCR tests complete for 08ebea6!

@SirGitsalot
SirGitsalot marked this pull request as ready for review August 13, 2026 00:58
@SirGitsalot
SirGitsalot requested a review from melinath August 13, 2026 01:00

@melinath melinath left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Just want to double-check that this has to do with regions? I didn't know that IAM permissions could differ between regions... also, if it's based on the configuration to that extent, why is this passing in beta?

@melinath

Copy link
Copy Markdown
Member

(I was wondering if this might be missing IAM bootstrapping: https://googlecloudplatform.github.io/magic-modules/test/test/#iam-resources)

@SirGitsalot

Copy link
Copy Markdown
Member Author

The region ends up in the name of the buckets, which are dataproc-[staging|temp]-[region]-[project number]-[seemingly random suffix]. The beta project has buckets for both us-central1 and asia-east1 while the ga project only has us-central1. The compute service accounts in both projects have the exact same set of roles, which makes me think its' the absence of the bucket rather than the absence of a role that's causing the failure.

The buckets should be created automatically, since this test doesn't specify custom buckets. If it's failing to do so in the GA project it doesn't appear to be due to IAM roles on the compute SA (or on the test runner SA, which between GA and beta does have a different set of roles, but both have "Owner" and "Storage Admin")

@github-actions

Copy link
Copy Markdown

@SirGitsalot, this PR is waiting for action from you. If no action is taken, this PR will be closed in 28 days.

Please address any comments or change requests, or re-request review from a core reviewer if no action is required.

Image showing the re-request review button

This notification can be disabled with the disable-automatic-closure label.

@SirGitsalot
SirGitsalot requested a review from melinath September 1, 2026 16:12
@github-actions

github-actions Bot commented Sep 4, 2026

Copy link
Copy Markdown

@melinath This PR has been waiting for review for 3 weekdays. Please take a look! Use the label disable-review-reminders to disable these notifications.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Failing test(s): TestAccDataprocCluster_AttachedDiskConfigWorker

3 participants